NO-ISSUE: Update github.com/openshift/hive/apis digest to 74185d6 - #10746
red-hat-konflux[bot] wants to merge 1 commit into
Conversation
|
@red-hat-konflux[bot]: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
1c20840 to
19f7d78
Compare
|
New changes are detected. LGTM label has been removed. |
19f7d78 to
7840990
Compare
Codecov Report✅ All modified and coverable lines are covered by tests. Please upload reports for the commit 16b42e0 to get more accurate results. Additional details and impacted files@@ Coverage Diff @@
## release-ocm-2.12 #10746 +/- ##
=================================================
Coverage 41.75% 41.75%
=================================================
Files 305 305
Lines 61921 61921
=================================================
Hits 25853 25853
Misses 33781 33781
Partials 2287 2287 🚀 New features to boost your workflow:
|
7840990 to
16b42e0
Compare
16b42e0 to
70b2a2c
Compare
4fa124a to
506a12d
Compare
506a12d to
23cf547
Compare
Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
23cf547 to
3d796f4
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: red-hat-konflux[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
@red-hat-konflux[bot]: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
This PR contains the following updates:
899e2c9→74185d6ad99b9e→74185d6Warning
Some dependencies could not be looked up. Check the warning logs for more information.
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.