fix: bump ts-jest from 29.4.12 to 29.4.14 - #474
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps [ts-jest](https://github.com/kulshekhar/ts-jest) from 29.4.12 to 29.4.14. - [Release notes](https://github.com/kulshekhar/ts-jest/releases) - [Changelog](https://github.com/kulshekhar/ts-jest/blob/main/CHANGELOG.md) - [Commits](kulshekhar/ts-jest@v29.4.12...v29.4.14) --- updated-dependencies: - dependency-name: ts-jest dependency-version: 29.4.14 dependency-type: direct:development update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
Bumps ts-jest from 29.4.12 to 29.4.14.
Release notes
Sourced from ts-jest's releases.
Changelog
Sourced from ts-jest's changelog.
Commits
2d152f6chore(release): 29.4.1401e5a2erevert: bring back prior changes of compiler util (#5471)8e35da6build(deps): update dependency typescript-eslint to ^8.70.176ec454build(deps): bump image-size from 2.0.2 to 2.0.4 in /websitea0c364fchore(release): 29.4.13697d337perf(compiler): cache module resolution modes (#5419)c3a0b57feat: support Babel 8 (#5466)5fb44c4build(deps): update dependency jest-environment-jsdom to ^30.5.2257a524build(deps): update dependency jest to v30.5.289ba4d9build(deps): update dependency babel-jest to ^30.5.2Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)