Skip to content

.zuul: update ansible-version to v11 - #1846

Merged
debarshiray merged 1 commit into
containers:mainfrom
TristanCacqueray:bump-ansible
Oct 8, 2026
Merged

debarshiray merged 1 commit into
containers:mainfrom
TristanCacqueray:bump-ansible

Conversation

@TristanCacqueray

Copy link
Copy Markdown
Contributor

This change bumps ansible-core to 2.18, which is now necessary for coreos-next. This is similar to #1824

Comment thread .zuul.yaml
name: system-test-fedora-coreos-next-commands-options
parent: system-test-fedora-coreos
description: Run Toolbx's commands-options system tests in Fedora CoreOS next
ansible-version: 11

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is there any downside to having this defined for all coreos jobs, not just the coreos-next ones? If not we could have just the 1 line in system-test-fedora-coreos

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I think that should be fine, but that's temporary, the next update Zuul brings ansible version 11 by default, see: https://zuul-ci.org/docs/zuul/latest/releasenotes.html#upgrade-notes

We recently updated our operator and we'll rollout the latest Zuul shortly ( https://softwarefactory-project.io/r/c/software-factory/sf-operator/+/35317 )

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/7a73d95a4c88421abf5e5ebca206b4c1

✔️ unit-test SUCCESS in 2m 23s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 22s
✔️ unit-test-restricted SUCCESS in 2m 22s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 45m 24s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 35m 09s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 8m 05s
✔️ system-test-fedora-44-commands-options SUCCESS in 35m 46s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 34m 08s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 6m 50s
✔️ system-test-fedora-43-commands-options SUCCESS in 43m 20s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 34m 29s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 6m 04s
✔️ system-test-fedora-42-commands-options SUCCESS in 42m 25s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 34m 32s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 8m 00s
❌ system-test-fedora-coreos-next-commands-options RETRY_LIMIT in 1m 02s
❌ system-test-fedora-coreos-next-runtime-environment-arch-fedora RETRY_LIMIT in 1m 03s
❌ system-test-fedora-coreos-next-runtime-environment-ubuntu RETRY_LIMIT in 1m 03s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 41m 31s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 34m 08s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 9m 28s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 42m 14s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 35m 13s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 9m 16s

@Rolv-Apneseth

Copy link
Copy Markdown
Contributor

And now looks like we have some packaging/repo errors.

Synchronize glibc to repo version for gcc/glibc-devel compatibility:

error: No matches for 'glibc' in repo 'updates'

Synchronize shadow-utils to repo version for shadow-utils-subid-devel compatibility:

error: No matches for 'shadow-utils' in repo 'updates'

Install RPM packages:

error: Could not depsolve transaction; 1 problem detected:
 Problem: libgcc-16.2.1-2.fc45.1.i686 from fedora does not belong to a distupgrade repository
  - package gcc-16.2.1-2.fc45.1.x86_64 from fedora requires libgcc >= 16.2.1-2.fc45.1, but none of the providers can be installed
  - cannot install both libgcc-16.2.1-2.fc45.1.x86_64 from fedora and libgcc-16.2.1-2.fc45.x86_64 from @System
  - conflicting requests

And a failure in ubuntu-tests:

not ok 184 network: DNS inside Arch Linux in 738ms
# tags: arch-fedora runtime-environment
# (from function `assert_line' in file test/system/libs/bats-assert/src/assert.bash, line 488,
#  in test file test/system/203-network.bats, line 283)
#   `assert_line --index 0 "$ipv4_addr"' failed
#
# -- line differs --
# index    : 0
# expected : 193.0.14.129
# actual   :
# --
#

@Rolv-Apneseth

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/5ec4d398abc34a0c92ddf2702a0ae2e1

✔️ unit-test SUCCESS in 2m 13s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 11s
✔️ unit-test-restricted SUCCESS in 2m 19s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 36m 34s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 29m 15s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 6m 47s
✔️ system-test-fedora-44-commands-options SUCCESS in 38m 03s
❌ system-test-fedora-44-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000225208 failed in 0s
❌ system-test-fedora-44-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000225202 failed in 0s
✔️ system-test-fedora-43-commands-options SUCCESS in 37m 25s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 28m 37s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 03s
✔️ system-test-fedora-42-commands-options SUCCESS in 36m 20s
❌ system-test-fedora-42-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000225203 failed in 0s
❌ system-test-fedora-42-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000225187 failed in 0s
❌ system-test-fedora-coreos-next-commands-options RETRY_LIMIT in 1m 03s
❌ system-test-fedora-coreos-next-runtime-environment-arch-fedora RETRY_LIMIT in 1m 03s
❌ system-test-fedora-coreos-next-runtime-environment-ubuntu RETRY_LIMIT in 1m 03s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 44m 52s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 36m 38s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 8m 00s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 45m 12s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 37m 49s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 10m 54s

@TristanCacqueray

Copy link
Copy Markdown
Contributor Author

The NODE_FAILURE are being fixed on our side, but the glibc error seems legit: looks like the coreos repos are missing key packages, or maybe we are missing a repofile?

This is similar to commit 93f882c and is needed because the
'next' stream of Fedora CoreOS was rebased to Fedora 45 [1].

[1] https://lists.fedoraproject.org/archives/list/coreos@lists.fedoraproject.org/thread/QG22WVLEPA5ES6XRAX2Q5JSUORFUJBRD/

containers#1846

Signed-off-by: Tristan Cacqueray <tdecacqu@redhat.com>

@debarshiray debarshiray left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for tracking this down, @TristanCacqueray ! I took the liberty to add the ID of the previous commit in the Git commit message.

@debarshiray

Copy link
Copy Markdown
Member

The NODE_FAILURE are being fixed on our side, but the glibc error seems legit: looks like the coreos repos are missing key packages, or maybe we are missing a repofile?

Let's get this merged once the CI finishes. It's a legit fix in itself, even if it doesn't solve everything.

@debarshiray

debarshiray commented Oct 8, 2026 •

Copy link
Copy Markdown
Member

And a failure in ubuntu-tests:

not ok 184 network: DNS inside Arch Linux in 738ms
# tags: arch-fedora runtime-environment
# (from function `assert_line' in file test/system/libs/bats-assert/src/assert.bash, line 488,
#  in test file test/system/203-network.bats, line 283)
#   `assert_line --index 0 "$ipv4_addr"' failed
#
# -- line differs --
# index    : 0
# expected : 193.0.14.129
# actual   :
# --
#

We used to see this once in a while in the past. I hadn't seen it in a while, and was beginning to hope that this was fixed by commit e13a58c76e64abc2. :(

@Rolv-Apneseth

Copy link
Copy Markdown
Contributor

The NODE_FAILURE are being fixed on our side, but the glibc error seems legit: looks like the coreos repos are missing key packages, or maybe we are missing a repofile?

Yes, sorry, I'm looking into it just haven't had much time. Using the updates-testing repo seems to work (see output on solvables below), so it's just something about how the repos for next are currently setup (I'm assuming because of branching). I'll get back to you once I have more details / understand it better as I'm not very familiar with that side of things.

rpm-md repo 'updates-testing' (cached); generated: 2026-10-05T01:50:59Z solvables: 12737
rpm-md repo 'updates' (cached); generated: 2018-02-20T19:18:14Z solvables: 0

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/8b59d98afe5d4ed9ac05cc1b1fade020

✔️ unit-test SUCCESS in 2m 13s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 28s
✔️ unit-test-restricted SUCCESS in 2m 03s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 37m 47s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 25m 48s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 6m 05s
✔️ system-test-fedora-44-commands-options SUCCESS in 40m 02s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 31m 25s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 6m 16s
✔️ system-test-fedora-43-commands-options SUCCESS in 37m 12s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 25m 51s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 06s
✔️ system-test-fedora-42-commands-options SUCCESS in 34m 16s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 26m 50s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 07s
❌ system-test-fedora-coreos-next-commands-options RETRY_LIMIT in 1m 03s
❌ system-test-fedora-coreos-next-runtime-environment-arch-fedora RETRY_LIMIT in 1m 03s
❌ system-test-fedora-coreos-next-runtime-environment-ubuntu RETRY_LIMIT in 1m 02s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 40m 04s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 31m 42s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 8m 34s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 44m 48s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 35m 37s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 9m 18s

@Rolv-Apneseth

Copy link
Copy Markdown
Contributor

The NODE_FAILURE are being fixed on our side, but the glibc error seems legit: looks like the coreos repos are missing key packages, or maybe we are missing a repofile?

Yes, sorry, I'm looking into it just haven't had much time. Using the updates-testing repo seems to work (see output on solvables below), so it's just something about how the repos for next are currently setup (I'm assuming because of branching). I'll get back to you once I have more details / understand it better as I'm not very familiar with that side of things.

rpm-md repo 'updates-testing' (cached); generated: 2026-10-05T01:50:59Z solvables: 12737
rpm-md repo 'updates' (cached); generated: 2018-02-20T19:18:14Z solvables: 0

Does seem like this is the issue. While next is pointing to F45, updates are all going to repo=updates-testing, so we need to pull packages from there. Working on a fix in #1847, but we should merge this first.

@debarshiray
debarshiray merged commit cdecacb into containers:main Oct 8, 2026
1 of 3 checks passed
@debarshiray

Copy link
Copy Markdown
Member

Thanks for all the effort, @Rolv-Apneseth @TristanCacqueray !

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants