Skip to content

Remove Trustpilot widget due to background color - #2508

Merged
MusabShakeel576 merged 1 commit into
masterfrom
trustpilot-widget
Sep 10, 2026
Merged

Remove Trustpilot widget due to background color#2508
MusabShakeel576 merged 1 commit into
masterfrom
trustpilot-widget

Conversation

@MusabShakeel576

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercel Bot commented Sep 10, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

2 Skipped Deployments
Project Deployment Actions Updated
solid-app Ignored Ignored Sep 10, 2026 2:21pm UTC
solid-app-staging Ignored Ignored Sep 10, 2026 2:21pm UTC

Request Review

@MusabShakeel576
MusabShakeel576 merged commit 70e9b37 into master Sep 10, 2026
5 checks passed
@MusabShakeel576
MusabShakeel576 deleted the trustpilot-widget branch September 10, 2026 14:21
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant